Skip to content

Sync with react.dev @ 7b6c3ceb - #725

Merged
anilcanboga merged 6 commits into
mainfrom
sync-7b6c3ceb
Sep 8, 2026
Merged

Sync with react.dev @ 7b6c3ceb#725
anilcanboga merged 6 commits into
mainfrom
sync-7b6c3ceb

Conversation

@react-translations-bot

@react-translations-bot react-translations-bot commented Jul 27, 2026

Copy link
Copy Markdown
Collaborator

This PR was automatically generated.

Merge changes from react.dev at 7b6c3ce

The following files have conflicts and may need new translations:

Please fix the conflicts by pushing new commits to this pull request, either by editing the files directly on GitHub or by checking out this branch.

DO NOT SQUASH MERGE THIS PULL REQUEST!

Doing so will "erase" the commits from main and cause them to show up as conflicts the next time we merge.

aurorascharff and others added 2 commits July 20, 2026 15:31
@vercel

vercel Bot commented Jul 27, 2026

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
tr-legacy-reactjs-org Ignored Ignored Preview Sep 8, 2026 8:01pm UTC

Request Review

@github-actions

github-actions Bot commented Sep 8, 2026

Copy link
Copy Markdown

Size changes

Details

📦 Next.js Bundle Analysis for react-dev

This analysis was generated by the Next.js Bundle Analysis action. 🤖

This PR introduced no changes to the JavaScript bundle! 🙌

@anilcanboga
anilcanboga merged commit 635638b into main Sep 8, 2026
10 checks passed
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants